Skip to content

chore: bump Helm and otelcontribcol - #2291

Open
Camila-B wants to merge 1 commit into
GoogleContainerTools:mainfrom
Camila-B:fix-vulns
Open

chore: bump Helm and otelcontribcol#2291
Camila-B wants to merge 1 commit into
GoogleContainerTools:mainfrom
Camila-B:fix-vulns

Conversation

@Camila-B

Copy link
Copy Markdown
Contributor
  • Bumps Helm to v3.21.1-gke.9
  • Bumps otelcontribcol to v0.152.0-gke.17

- Bumps Helm to v3.21.1-gke.9
- Bumps otelcontribcol to v0.152.0-gke.17
@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please ask for approval from camila-b. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant