chore: update SDK from api-definitions - #793
Open
ghost wants to merge 1 commit into
Open
Conversation
ghost
force-pushed
the
automations/update-sdk-from-api-definitions
branch
from
June 16, 2026 08:50
c5b9004 to
73e878d
Compare
ghost
self-requested a review
June 16, 2026 08:50
🦋 Changeset detectedLatest commit: ad10ac2 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
ghost
force-pushed
the
automations/update-sdk-from-api-definitions
branch
9 times, most recently
from
June 23, 2026 16:17
6fec251 to
58d773b
Compare
ghost
force-pushed
the
automations/update-sdk-from-api-definitions
branch
3 times, most recently
from
June 25, 2026 14:14
b614712 to
de72bbf
Compare
rebilly-ci-sdk-write
Bot
force-pushed
the
automations/update-sdk-from-api-definitions
branch
from
September 3, 2026 10:03
de72bbf to
9d614de
Compare
rebilly-ci-sdk-write
Bot
force-pushed
the
automations/update-sdk-from-api-definitions
branch
11 times, most recently
from
September 11, 2026 12:08
d3b7c57 to
ab10fe5
Compare
rebilly-ci-sdk-write
Bot
force-pushed
the
automations/update-sdk-from-api-definitions
branch
from
September 11, 2026 12:51
ab10fe5 to
ad10ac2
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR is automatically generated and updates SDK based on a change in the API definitions