Skip to content

Add new midstream branches and update pipelines - #1018

Open
serverless-qe wants to merge 1 commit into
mainfrom
discover-serverless-ci
Open

serverless-qe wants to merge 1 commit into
mainfrom
discover-serverless-ci

Conversation

@serverless-qe

Copy link
Copy Markdown
Collaborator

Discover new branches and update Konflux pipelines using openshift-knative/hack.

@openshift-ci
openshift-ci Bot requested review from aliok and creydr September 21, 2026 09:24
@openshift-ci

openshift-ci Bot commented Sep 21, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign matzew for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@Kaustubh-pande

Copy link
Copy Markdown
Collaborator

/hold

@serverless-qe
serverless-qe force-pushed the discover-serverless-ci branch 5 times, most recently from b11aee6 to 3d8c24c Compare September 28, 2026 10:05

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants