Skip to content

Remove matzew and creydr as reviewers and approvers - #1026

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-knative:mainfrom
creydr:remove-matzew-and-myself-as-reviewers
Sep 29, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift-knative:mainfrom
creydr:remove-matzew-and-myself-as-reviewers

Conversation

@creydr

@creydr creydr commented Sep 29, 2026

Copy link
Copy Markdown
Collaborator

As we're not working on those topics ATM.

@matzew: fyi

@matzew matzew left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

/ok-to-test

@openshift-ci

openshift-ci Bot commented Sep 29, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: creydr, matzew

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit dddb284 into openshift-knative:main Sep 29, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants