chore(deps): bump oas-normalize from 16.1.2 to 18.0.2 - #1204
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 7266005 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
erunion
removed their request for review
September 2, 2026 21:59
dependabot
Bot
force-pushed
the
dependabot/npm_and_yarn/oas-normalize-18.0.0
branch
from
September 11, 2026 20:45
b7645cd to
78b6c25
Compare
Bumps [oas-normalize](https://github.com/readmeio/oas/tree/HEAD/packages/oas-normalize) from 16.1.2 to 18.0.2. - [Release notes](https://github.com/readmeio/oas/releases) - [Changelog](https://github.com/readmeio/oas/blob/main/packages/oas-normalize/CHANGELOG.md) - [Commits](https://github.com/readmeio/oas/commits/18.0.2/packages/oas-normalize) --- updated-dependencies: - dependency-name: oas-normalize dependency-version: 18.0.0 dependency-type: direct:production update-type: version-update:semver-major ... Signed-off-by: dependabot[bot] <support@github.com>
dependabot
Bot
force-pushed
the
dependabot/npm_and_yarn/oas-normalize-18.0.0
branch
from
September 11, 2026 20:54
78b6c25 to
cd3afaa
Compare
erunion
approved these changes
Sep 11, 2026
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Bumps oas-normalize from 16.1.2 to 18.0.2.
Release notes
Sourced from oas-normalize's releases.
... (truncated)
Changelog
Sourced from oas-normalize's changelog.
Commits