Skip to content

migrate mcm-provider from openstack to stackit - #272

Open
aniruddha2000 wants to merge 17 commits into
mainfrom
migrate-mcm
Open

migrate mcm-provider from openstack to stackit#272
aniruddha2000 wants to merge 17 commits into
mainfrom
migrate-mcm

Conversation

@aniruddha2000

@aniruddha2000 aniruddha2000 commented Aug 26, 2026

Copy link
Copy Markdown
Member

How to categorize this PR?

/kind enhancement

What this PR does / why we need it:

This PR replaces mcm provider from openstack to stackit. Replaces openstack provider ID with stackit provider ID and add stackit.cloud/migrated-machine: true annotation to machines and puts annotation stackit.cloud/machine-controller-manager-migrated: true to workers once the migration is done.

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

@ske-prow ske-prow Bot added the kind/enhancement Enhancement, improvement, extension label Aug 26, 2026
@ske-prow

ske-prow Bot commented Aug 26, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign aniruddha2000 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Aug 26, 2026
@ske-prow ske-prow Bot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Aug 28, 2026
@ske-prow ske-prow Bot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Aug 31, 2026
@aniruddha2000

Copy link
Copy Markdown
Member Author

/retest-required

@aniruddha2000

Copy link
Copy Markdown
Member Author

/retest

3 similar comments
@aniruddha2000

Copy link
Copy Markdown
Member Author

/retest

@aniruddha2000

Copy link
Copy Markdown
Member Author

/retest

@aniruddha2000

Copy link
Copy Markdown
Member Author

/retest

@aniruddha2000

Copy link
Copy Markdown
Member Author

/test pull-gardener-extension-provider-stackit-dev-artifacts

@aniruddha2000

Copy link
Copy Markdown
Member Author

/test all

@aniruddha2000

Copy link
Copy Markdown
Member Author

/test pull-gardener-extension-provider-stackit-integration-tests

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/enhancement Enhancement, improvement, extension size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant