Skip to content

feat: use cloud profile storage type and size as defaults for machineclasses - #290

Open
nschad wants to merge 2 commits into
mainfrom
add-default-storageclass
Open

feat: use cloud profile storage type and size as defaults for machineclasses#290
nschad wants to merge 2 commits into
mainfrom
add-default-storageclass

Conversation

@nschad

@nschad nschad commented Sep 7, 2026

Copy link
Copy Markdown
Contributor

How to categorize this PR?

/kind enhancement

What this PR does / why we need it:

See Title, pretty self explanatory. Add's better defaulting if not set especially in MachineClass

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

… classes

Signed-off-by: Niclas Schad <niclas.schad@stackit.cloud>
@ske-prow ske-prow Bot added the kind/enhancement Enhancement, improvement, extension label Sep 7, 2026
@ske-prow

ske-prow Bot commented Sep 7, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign ftl for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Sep 7, 2026
Signed-off-by: Niclas Schad <niclas.schad@stackit.cloud>
@ske-prow

ske-prow Bot commented Sep 7, 2026

Copy link
Copy Markdown

@nschad: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-gardener-extension-provider-stackit-integration-tests 00ebf96 link true /test pull-gardener-extension-provider-stackit-integration-tests

Full PR test history. Your PR dashboard. Command help for this repository.
Please help us cut down on flakes by linking this test failure to an open flake report or filing a new flake report if you can't find an existing one. Also see the gardener testing guideline for how to avoid and hunt flakes.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@ftl

ftl commented Sep 7, 2026

Copy link
Copy Markdown
Contributor

/lgtm

@ske-prow ske-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 7, 2026
@ske-prow

ske-prow Bot commented Sep 7, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 43e5f8250b77d707cc8a89f369e6903edd6b90dd

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/enhancement Enhancement, improvement, extension lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants