Skip to content

Make floatingPoolName optional - #280

Merged
ske-prow[bot] merged 1 commit into
mainfrom
makeFloatingPoolNameOptional
Sep 3, 2026
Merged

Make floatingPoolName optional#280
ske-prow[bot] merged 1 commit into
mainfrom
makeFloatingPoolNameOptional

Conversation

@dergeberl

@dergeberl dergeberl commented Sep 2, 2026

Copy link
Copy Markdown
Member

How to categorize this PR?

/kind enhancement
/cc @stackitcloud/ske-infrastructure

What this PR does / why we need it:
The floatingPoolName was required but never used in a STACKIT only Cluster (since the validation is in place #212). It is only used in the STACKIT Infra-Controller in case there are OpenStack Credentials to pull the IDs of it in the Status.

This PR removes the validation (we know we need it for clusters with OpenStack). To not make the unrelevant field for STACKIT Only clusters required.

There will be a followup PR (#281) which also further cleans up the floatingPoolName and other fields in the InfrastructureConfig.

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

@ske-prow
ske-prow Bot requested a review from a team September 2, 2026 15:26
@ske-prow ske-prow Bot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. kind/enhancement Enhancement, improvement, extension size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 2, 2026
@dergeberl
dergeberl force-pushed the makeFloatingPoolNameOptional branch from 44fa5ed to c1a39c4 Compare September 3, 2026 05:52
@ske-prow ske-prow Bot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 3, 2026
@dergeberl

Copy link
Copy Markdown
Member Author

/kind bug

@ske-prow ske-prow Bot added the kind/bug Bug label Sep 3, 2026
@dergeberl dergeberl changed the title WIP: Make floatingPoolName optional Make floatingPoolName optional Sep 3, 2026
@ske-prow ske-prow Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 3, 2026
@ske-prow ske-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 3, 2026
@ske-prow

ske-prow Bot commented Sep 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: nschad

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow

ske-prow Bot commented Sep 3, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 777d383fb719a4b97dcc38d917671d5e51744534

@ske-prow ske-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 3, 2026
@nschad

nschad commented Sep 3, 2026

Copy link
Copy Markdown
Contributor

/retest

@ske-prow
ske-prow Bot merged commit 0a7e641 into main Sep 3, 2026
4 checks passed
@ske-prow
ske-prow Bot deleted the makeFloatingPoolNameOptional branch September 3, 2026 08:02
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/bug Bug kind/enhancement Enhancement, improvement, extension lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants