Skip to content

Deprecate OpenStack only fields in InfrastructureConfig - #281

Open
dergeberl wants to merge 4 commits into
mainfrom
deprecateOpenStackOnlyFieldsInInfrastructureConfig
Open

Deprecate OpenStack only fields in InfrastructureConfig#281
dergeberl wants to merge 4 commits into
mainfrom
deprecateOpenStackOnlyFieldsInInfrastructureConfig

Conversation

@dergeberl

@dergeberl dergeberl commented Sep 3, 2026

Copy link
Copy Markdown
Member

How to categorize this PR?

/kind enhancement
/cc @stackitcloud/ske-infrastructure

What this PR does / why we need it:
This is a followup from #280.

The floatingPoolName was required but never used in a STACKIT only Cluster (since the validation is in place #212). It is only used in the STACKIT Infra-Controller in case there are OpenStack Credentials to pull the IDs of it in the Status.

The #280 PR removed the validation. This PR removes that the FloatingNetID is get via OpenStack in the STACKIT Infra-Controller (even if there are OpenStack Credenitals), as there is no need for it. This PR sets all only OpenStack fields in the Infrastructure to Depreciated. In OpenStack we save the EgressIPs in the provider status of the Infrastructure object to put it later in the Infrastructure Object. This is now done directly via context for STACKIT to not use the unneeded deprecated field in STACKIT Infra Controller.

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Breaking changes:

@ske-prow
ske-prow Bot requested a review from a team September 3, 2026 05:55
@ske-prow ske-prow Bot added kind/enhancement Enhancement, improvement, extension do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. labels Sep 3, 2026
@ske-prow

ske-prow Bot commented Sep 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign aniruddha2000 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 3, 2026
@dergeberl
dergeberl force-pushed the deprecateOpenStackOnlyFieldsInInfrastructureConfig branch from 7e373d4 to a5ce9dc Compare September 3, 2026 08:34
Use exclude in golangci-lint instead of nolint for each
@dergeberl
dergeberl force-pushed the deprecateOpenStackOnlyFieldsInInfrastructureConfig branch from a5ce9dc to a799000 Compare September 3, 2026 08:55
@dergeberl dergeberl changed the title WIP: Deprecate open stack only fields in infrastructure config WIP: Deprecate OpenStack only fields in InfrastructureConfig Sep 3, 2026
@dergeberl
dergeberl force-pushed the deprecateOpenStackOnlyFieldsInInfrastructureConfig branch from a799000 to dfa19ac Compare September 3, 2026 09:21
@dergeberl

Copy link
Copy Markdown
Member Author

/hold
needs testing in for different scenarios:

  • Openstack Infra-controller
  • STACKIT Infra-Controller
  • STACKIT Infra-Controller (without OpenStack credentials)

@ske-prow ske-prow Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 4, 2026
@dergeberl dergeberl changed the title WIP: Deprecate OpenStack only fields in InfrastructureConfig Deprecate OpenStack only fields in InfrastructureConfig Sep 4, 2026
@ske-prow ske-prow Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 4, 2026
IP string `json:"ip"`
// ExternalFixedIPs is the list of the router's assigned external fixed IPs.
//
// Deprecated: OpenStack-only; not used for STACKIT.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Isn't this used in other extensions?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Other extensions should depend on the infra.Status.EgressCIDRs (They should not depend on the the provider part). Do you have anything special in mind?

Example for acl:
https://github.com/stackitcloud/gardener-extension-acl/blob/33491918ea575b3241d4c594d80ea517d9d80670/pkg/controller/actuator.go#L191

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. kind/enhancement Enhancement, improvement, extension size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants